Skip to content

Add CODEOWNERS file - #19

Merged
dhmlau merged 1 commit into
masterfrom
add-codeowner
Aug 16, 2017
Merged

dhmlau merged 1 commit into
masterfrom
add-codeowner

Conversation

@dhmlau

@dhmlau dhmlau commented Aug 16, 2017

Copy link
Copy Markdown
Member

connect to strongloop-internal/scrum-apex#278

@dhmlau dhmlau added the #review label Aug 16, 2017
@dhmlau
dhmlau merged commit 9352b44 into master Aug 16, 2017
@dhmlau
dhmlau deleted the add-codeowner branch August 16, 2017 20:41
@dhmlau dhmlau removed the #review label Aug 16, 2017
@dhmlau dhmlau self-assigned this Aug 16, 2017
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants