chore: update full page screenshots - #8444
Conversation
|
Important Review skippedReview was skipped due to path filters ⛔ Files ignored due to path filters (118)
CodeRabbit blocks several paths by default. You can override this behavior by explicitly including those paths in the path filters. For example, including ⚙️ Run configurationConfiguration used: Repository UI (base), Organization UI (inherited) Review profile: CHILL Plan: Pro Run ID: You can disable this status message by setting the Use the checkbox below for a quick retry:
Thanks for using CodeRabbit! It's free for OSS, and your support helps us grow. If you like it, consider giving us a shout-out. Comment |
|
Preview: https://pf-pr-8444.surge.sh A11y report: https://pf-pr-8444-a11y.surge.sh |
|
🎉 This PR is included in version 6.6.0-prerelease.7 🎉 The release is available on: Your semantic-release bot 📦🚀 |
fixes #8392