chore: update backstop screenshots - #8362
Conversation
|
Important Review skippedReview was skipped due to path filters ⛔ Files ignored due to path filters (108)
CodeRabbit blocks several paths by default. You can override this behavior by explicitly including those paths in the path filters. For example, including ⚙️ Run configurationConfiguration used: Repository UI (base), Organization UI (inherited) Review profile: CHILL Plan: Pro Run ID: You can disable this status message by setting the Use the checkbox below for a quick retry:
Thanks for using CodeRabbit! It's free for OSS, and your support helps us grow. If you like it, consider giving us a shout-out. Comment |
|
Preview: https://pf-pr-8362.surge.sh A11y report: https://pf-pr-8362-a11y.surge.sh |
|
🎉 This PR is included in version 6.5.0-prerelease.82 🎉 The release is available on: Your semantic-release bot 📦🚀 |
Updates main and glass screenshots. Everything looks expected except it looks like there is a bug in pagination on mobile in glass. The sticky background isn't applying. You can see it in a bunch of examples, like
/components/table/html-demos/compactand/components/table/html-demos/compactglass.pdf
BackstopJS Report.pdf