chore(regression tests): update baseline with q4 work - #8039
Conversation
|
Important Review skippedReview was skipped due to path filters ⛔ Files ignored due to path filters (300)
CodeRabbit blocks several paths by default. You can override this behavior by explicitly including those paths in the path filters. For example, including You can disable this status message by setting the Comment |
|
Preview: https://pf-pr-8039.surge.sh A11y report: https://pf-pr-8039-a11y.surge.sh |
|
Found one bug, opened a separate issue for it - #8044 |
|
🎉 This PR is included in version 6.5.0-prerelease.34 🎉 The release is available on: Your semantic-release bot 📦🚀 |
Updates the baseline visual regression tests. Changes noted include:
Full report