Skip to content

meta: Add CODEOWNERS - #4182

Merged
sentrivana merged 1 commit into
masterfrom
ivana/add-codeowners
Mar 21, 2025
Merged

sentrivana merged 1 commit into
masterfrom
ivana/add-codeowners

Conversation

@sentrivana

@sentrivana sentrivana commented Mar 21, 2025 •

Copy link
Copy Markdown
Contributor

Ref #4183

@sentrivana sentrivana changed the title Add CODEOWNERS meta: Add CODEOWNERS Mar 21, 2025
@sentrivana
sentrivana requested a review from a team March 21, 2025 09:21
@codecov

codecov Bot commented Mar 21, 2025 •

Copy link
Copy Markdown

Codecov Report

All modified and coverable lines are covered by tests ✅

Project coverage is 79.57%. Comparing base (12b3ca3) to head (3ac5c20).
Report is 1 commits behind head on master.

✅ All tests successful. No failed tests found.

Additional details and impacted files
@@            Coverage Diff             @@
##           master    #4182      +/-   ##
==========================================
+ Coverage   79.56%   79.57%   +0.01%     
==========================================
  Files         141      141              
  Lines       15736    15736              
  Branches     2675     2675              
==========================================
+ Hits        12520    12522       +2     
  Misses       2369     2369              
+ Partials      847      845       -2     

see 4 files with indirect coverage changes

@sentrivana
sentrivana merged commit c295047 into master Mar 21, 2025
@sentrivana
sentrivana deleted the ivana/add-codeowners branch March 21, 2025 09:30
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants