5️⃣ <OrganizationProfile/> List, add, edit, remove Domains - #1560
Conversation
🦋 Changeset detectedLatest commit: 3dc622a The changes in this PR will be included in the next version bump. This PR includes changesets to release 12 packages
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
|
!snapshot |
|
Hey @panteliselef - the snapshot version command generated the following package versions:
Tip: use the snippet copy button below to quickly install the required packages. # @clerk/backend
npm i @clerk/[email protected]# @clerk/chrome-extension
npm i @clerk/[email protected]# @clerk/clerk-js
npm i @clerk/[email protected]# eslint-config-custom
npm i [email protected]# @clerk/clerk-expo
npm i @clerk/[email protected]# @clerk/fastify
npm i @clerk/[email protected]# gatsby-plugin-clerk
npm i [email protected]# @clerk/localizations
npm i @clerk/[email protected]# @clerk/nextjs
npm i @clerk/[email protected]# @clerk/clerk-react
npm i @clerk/[email protected]# @clerk/remix
npm i @clerk/[email protected]# @clerk/clerk-sdk-node
npm i @clerk/[email protected]# @clerk/shared
npm i @clerk/[email protected]# @clerk/themes
npm i @clerk/[email protected]# @clerk/types
npm i @clerk/[email protected] |
3640ba7 to
43d3432
Compare
| return []; | ||
| } | ||
|
|
||
| return domains.data.filter(d => { |
There was a problem hiding this comment.
❓ Why do we need to filter the domainList? Shouldn't FAPI take care of that?
There was a problem hiding this comment.
Yes, FAPI should be in charge of that.
At that moment filtering not available tho. A following PR aims to fix that.
| ); | ||
| }; | ||
|
|
||
| const OrganizationDomainsSection = () => { |
There was a problem hiding this comment.
🔧 Consider moving this to a separate file.
…t it + New `/add-domain/` route + New `/verify-domain/:id` route + New `/verified-domain/:id` route
+ Move RemoveResourcePage under common
This reverts commit b73df88.
|
This PR has been automatically locked since there has not been any recent activity after it was closed. Please open a new issue for related bugs. |
Type of change
Packages affected
@clerk/clerk-js@clerk/clerk-react@clerk/nextjs@clerk/remix@clerk/types@clerk/themes@clerk/localizations@clerk/clerk-expo@clerk/backend@clerk/clerk-sdk-node@clerk/shared@clerk/fastify@clerk/chrome-extensiongatsby-plugin-clerkbuild/tooling/choreDescription
npm testruns as expected.npm run buildruns as expected.Tests will be added in a following PR
🚨This PR adds
<2 KBin the organizationprofile.js chunk 🚨This is not by accident
Video
screen-recording-2023-08-10-at-30745-pm_eT1QcrdM.mp4