Skip to content

fix(backend): Add missing deleteSelfEnabled property in UpdateUserParams - #4044

Merged
LekoArts merged 1 commit into
mainfrom
elef/sdki-628-add-deleteselfenabled-parameter-to-updateuserparams
Aug 29, 2024
Merged

fix(backend): Add missing deleteSelfEnabled property in UpdateUserParams#4044
LekoArts merged 1 commit into
mainfrom
elef/sdki-628-add-deleteselfenabled-parameter-to-updateuserparams

Conversation

@panteliselef

Copy link
Copy Markdown
Contributor

Description

deleteSelfEnabled was missing from the UpdateUserParams type

Checklist

  • npm test runs as expected.
  • npm run build runs as expected.
  • (If applicable) JSDoc comments have been added or updated for any package exports
  • (If applicable) Documentation has been updated

Type of change

  • 🐛 Bug fix
  • 🌟 New feature
  • 🔨 Breaking change
  • 📖 Refactoring / dependency upgrade / documentation
  • other:

@panteliselef panteliselef self-assigned this Aug 28, 2024
@changeset-bot

changeset-bot Bot commented Aug 28, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 61b1f16

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 9 packages
Name Type
@clerk/backend Minor
@clerk/astro Patch
@clerk/express Patch
@clerk/fastify Patch
@clerk/nextjs Patch
@clerk/remix Patch
@clerk/clerk-sdk-node Patch
@clerk/tanstack-start Patch
@clerk/testing Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@panteliselef
panteliselef requested a review from a team August 28, 2024 16:55
@panteliselef

Copy link
Copy Markdown
Contributor Author

!snapshot

@clerk-cookie

Copy link
Copy Markdown
Collaborator

Hey @panteliselef - the snapshot version command generated the following package versions:

Package Version
@clerk/astro 1.2.4-snapshot.v61b1f16
@clerk/backend 1.9.0-snapshot.v61b1f16
@clerk/express 0.0.34-snapshot.v61b1f16
@clerk/fastify 1.0.36-snapshot.v61b1f16
@clerk/nextjs 5.3.7-snapshot.v61b1f16
@clerk/remix 4.2.20-snapshot.v61b1f16
@clerk/clerk-sdk-node 5.0.33-snapshot.v61b1f16
@clerk/tanstack-start 0.3.3-snapshot.v61b1f16
@clerk/testing 1.2.16-snapshot.v61b1f16

Tip: Use the snippet copy button below to quickly install the required packages.
@clerk/astro

npm i @clerk/[email protected] --save-exact

@clerk/backend

npm i @clerk/[email protected] --save-exact

@clerk/express

npm i @clerk/[email protected] --save-exact

@clerk/fastify

npm i @clerk/[email protected] --save-exact

@clerk/nextjs

npm i @clerk/[email protected] --save-exact

@clerk/remix

npm i @clerk/[email protected] --save-exact

@clerk/clerk-sdk-node

npm i @clerk/[email protected] --save-exact

@clerk/tanstack-start

npm i @clerk/[email protected] --save-exact

@clerk/testing

npm i @clerk/[email protected] --save-exact

@LekoArts
LekoArts merged commit 9d04777 into main Aug 29, 2024
@LekoArts
LekoArts deleted the elef/sdki-628-add-deleteselfenabled-parameter-to-updateuserparams branch August 29, 2024 06:07
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants