Skip to content

fix(ui): Font inconsistencies - #3834

Merged
joe-bell merged 2 commits into
joe-bell/SDKI-496from
joe-bell/SDKI-504
Jul 29, 2024
Merged

fix(ui): Font inconsistencies#3834
joe-bell merged 2 commits into
joe-bell/SDKI-496from
joe-bell/SDKI-504

Conversation

@joe-bell

Copy link
Copy Markdown
Contributor

Description

Most issues turn out to be fixed by #3820 (Inter made it tricky to identify inconsistencies)

  • Fix separator spacing
  • Adjust Card footer weight
  • Use a disabled LinkButton within the resend for a more consistent experience (additionally, a disabled button provides better context on why the content has a countdown)

Checklist

  • npm test runs as expected.
  • npm run build runs as expected.
  • (If applicable) JSDoc comments have been added or updated for any package exports
  • (If applicable) Documentation has been updated

Type of change

  • 🐛 Bug fix
  • 🌟 New feature
  • 🔨 Breaking change
  • 📖 Refactoring / dependency upgrade / documentation
  • other:

@changeset-bot

changeset-bot Bot commented Jul 29, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: e363f36

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 0 packages

When changesets are added to this PR, you'll see the packages that this PR includes changesets for and the associated semver types

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@joe-bell
joe-bell merged commit 01ce74e into joe-bell/SDKI-496 Jul 29, 2024
@joe-bell
joe-bell deleted the joe-bell/SDKI-504 branch July 29, 2024 12:00
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants