Skip to content

refactor(ui): Use current accent color, switch to system font, address font inconsistencies - #3820

Merged
joe-bell merged 4 commits into
mainfrom
joe-bell/SDKI-496
Jul 29, 2024
Merged

refactor(ui): Use current accent color, switch to system font, address font inconsistencies#3820
joe-bell merged 4 commits into
mainfrom
joe-bell/SDKI-496

Conversation

@joe-bell

@joe-bell joe-bell commented Jul 25, 2024

Copy link
Copy Markdown
Contributor

Description

Note

This PR also contains #3834

Caught up with Alvish today and he made a few recommendations that were relatively quick to fix:

  1. Use our current (purple) accent color to reduce confusion between our accent and grayscale palette
  2. Default to system font – current behaviour – rather than inherit

Setting the --cl-font-family has no effect currently as we haven't yet decided how to approach "global" styles within components, so for now I'm going to simply remove Inter (we can reassess later if needed)

Before

CleanShot 2024-07-25 at 16 23 22@2x

After

CleanShot 2024-07-25 at 16 23 01@2x

Checklist

  • npm test runs as expected.
  • npm run build runs as expected.
  • (If applicable) JSDoc comments have been added or updated for any package exports
  • (If applicable) Documentation has been updated

Type of change

  • 🐛 Bug fix
  • 🌟 New feature
  • 🔨 Breaking change
  • 📖 Refactoring / dependency upgrade / documentation
  • other:

@changeset-bot

changeset-bot Bot commented Jul 25, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 01ce74e

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 0 packages

When changesets are added to this PR, you'll see the packages that this PR includes changesets for and the associated semver types

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@joe-bell
joe-bell requested a review from alexcarpenter July 26, 2024 08:11
@joe-bell
joe-bell force-pushed the joe-bell/SDKI-496 branch from 9f2371c to cf39c3e Compare July 29, 2024 10:49
@joe-bell joe-bell mentioned this pull request Jul 29, 2024
12 tasks
@joe-bell joe-bell changed the title refactor(ui): Use current accent color, switch to system font refactor(ui): Use current accent color, switch to system font, address font inconsistencies Jul 29, 2024

@alexcarpenter alexcarpenter left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Sorry I missed the review here on friday.

I would not assume we'd update the primary accent color to be our purple but something we can circle back on and not a blocker.

import { ToggleGroup } from './toggle-group';

const lightAccentDefault = '#2F3037';
const lightAccentDefault = '#6C47FF';

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

I would assume we'd not want our default accent color to be the clerk purple and default to the current gray.

Copy link
Copy Markdown
Contributor Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

No worries on the delay!

I went with Alvish's recommendation for now to help distinguish between the gray/accent (which I agree has proven tricky to differentiate at times)

@joe-bell
joe-bell merged commit fdd3ed1 into main Jul 29, 2024
@joe-bell
joe-bell deleted the joe-bell/SDKI-496 branch July 29, 2024 13:18
brkalow pushed a commit that referenced this pull request Aug 1, 2024
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants