Skip to content

refactor(ui): Extract useEnabledConnections hook - #3619

Merged
alexcarpenter merged 2 commits into
mainfrom
alexcarpenter/refactor-use-enabled-connections-hook
Jun 21, 2024
Merged

refactor(ui): Extract useEnabledConnections hook#3619
alexcarpenter merged 2 commits into
mainfrom
alexcarpenter/refactor-use-enabled-connections-hook

Conversation

@alexcarpenter

Copy link
Copy Markdown
Member

Description

Extract useEnabledConnections and remove getEnabledSocialConnectionsFromEnvironment util

Checklist

  • npm test runs as expected.
  • npm run build runs as expected.
  • (If applicable) JSDoc comments have been added or updated for any package exports
  • (If applicable) Documentation has been updated

Type of change

  • 🐛 Bug fix
  • 🌟 New feature
  • 🔨 Breaking change
  • 📖 Refactoring / dependency upgrade / documentation
  • other:

@changeset-bot

changeset-bot Bot commented Jun 21, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: b16a5e6

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 0 packages

When changesets are added to this PR, you'll see the packages that this PR includes changesets for and the associated semver types

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@alexcarpenter
alexcarpenter requested a review from tmilewski June 21, 2024 21:02
@alexcarpenter
alexcarpenter enabled auto-merge (squash) June 21, 2024 21:05
@alexcarpenter
alexcarpenter disabled auto-merge June 21, 2024 21:08
@alexcarpenter
alexcarpenter enabled auto-merge (squash) June 21, 2024 21:09
@alexcarpenter
alexcarpenter merged commit a94de32 into main Jun 21, 2024
@alexcarpenter
alexcarpenter deleted the alexcarpenter/refactor-use-enabled-connections-hook branch June 21, 2024 21:11
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants