Skip to content

ci: Migrate from CODEOWNERS to PullApprove for PR review management - #34814

Closed
josephperrott wants to merge 12 commits into
angular:masterfrom
josephperrott:pullapprove
Closed

josephperrott wants to merge 12 commits into
angular:masterfrom
josephperrott:pullapprove

Conversation

@josephperrott

Copy link
Copy Markdown
Member

We are migrating to PullApprove for our PR review management in an attempt
to allow for more granular and equitable code review assignments across the
team. Currently this migration is equivalent in the review assignments
it will create. Once stable, our expectation is that we will be able to
take advantage of PullApprove's additional features for things like staged
reviews.

@josephperrott josephperrott added area: build & ci Related the build and CI infrastructure of the project target: patch This PR is targeted for the next patch release labels Jan 16, 2020
@ngbot ngbot Bot added this to the needsTriage milestone Jan 16, 2020
@josephperrott
josephperrott marked this pull request as ready for review January 16, 2020 18:00
@josephperrott
josephperrott requested a review from a team January 16, 2020 18:00
Comment thread .pullapprove.yml Outdated
Comment thread .circleci/config.yml Outdated

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

😞 Shouldn't we have a similar check for pullapprove?

Copy link
Copy Markdown
Member Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Yes, The plan is to do this logic/check in a follow up PR.

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Is there a tracking item in our backlog for this followup pr?

Comment thread .pullapprove.yml Outdated

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

I have gone missing from this list for some reason (I'm still in the compiler group below)

Copy link
Copy Markdown
Member Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Sorry! I used the listing at aio/content/marketing/contributors.json to create the list and must have accidentally dropped you out of the list.

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Does this include collaborators (noticed JiaLi in the list)? If so, I'm missing too!

Copy link
Copy Markdown
Member Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

The intention is for it to be a list that contains all of the usernames that are present as an owner for any of the groups defined in the file.

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

@IgorMinar @josephperrott Something I’ve been wondering for a while then, can we add the Googlers and Collaborators for Universal as members on platform-server? (That would be me, @alan-agius4, @kyliau, and @manfredsteyer)

@googlebot

Copy link
Copy Markdown

All (the pull request submitter and all commit authors) CLAs are signed, but one or more commits were authored or co-authored by someone other than the pull request submitter.

We need to confirm that all authors are ok with their commits being contributed to this project. Please have them confirm that by leaving a comment that contains only @googlebot I consent. in this pull request.

Note to project maintainer: There may be cases where the author cannot leave a comment, or the comment is not properly detected as consent. In those cases, you can manually confirm consent of the commit author(s), and set the cla label to yes (if enabled on your project).

ℹ️ Googlers: Go here for more info.

@IgorMinar IgorMinar left a comment •

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Posting comments from my first pass... I also pushed a commit with some fixes I made locally.

Things we should not forget to update:

  • ng-bot config to require pullapprove status (.github/angular-robot.yml)
  • docs/COMMITTER.md which makes references to codeowners
  • docs/TRIAGE_AND_LABELS.md which also makes references to codeowners
  • what to do with existing gh groups? delete them?

Comment thread .circleci/config.yml Outdated

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Is there a tracking item in our backlog for this followup pr?

Comment thread .pullapprove.yml Outdated

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

contains_any_globs is an odd name. shouldn't it be matches_any_globs?

also what is files in this context? the files modified by the PR?

Copy link
Copy Markdown
Member Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

matches_any_globs probably makes more sense, but is not the name of the method that they went with. I think it comes from a consistency in naming conventions with there other matcher methods.

Yes, all of the files modified in the PR.

Comment thread .pullapprove.yml Outdated

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

can you leave a comment explaining how this works? why active groups?

@IgorMinar

Copy link
Copy Markdown
Contributor

@josephperrott can you please review my commit + rebase this? thanks

@josephperrott josephperrott left a comment

Copy link
Copy Markdown
Member Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Created #34956 for the verify-codeownership followup.

@josephperrott
josephperrott force-pushed the pullapprove branch 2 times, most recently from d767544 to 9e142e0 Compare January 24, 2020 20:16
@googlebot

Copy link
Copy Markdown

A Googler has manually verified that the CLAs look good.

(Googler, please make sure the reason for overriding the CLA status is clearly documented in these comments.)

ℹ️ Googlers: Go here for more info.

@googlebot

Copy link
Copy Markdown

All (the pull request submitter and all commit authors) CLAs are signed, but one or more commits were authored or co-authored by someone other than the pull request submitter.

We need to confirm that all authors are ok with their commits being contributed to this project. Please have them confirm that by leaving a comment that contains only @googlebot I consent. in this pull request.

Note to project maintainer: There may be cases where the author cannot leave a comment, or the comment is not properly detected as consent. In those cases, you can manually confirm consent of the commit author(s), and set the cla label to yes (if enabled on your project).

ℹ️ Googlers: Go here for more info.

josephperrott and others added 3 commits January 27, 2020 18:37
We are migrating to PullApprove for our PR review management in an attempt
to allow for more granular and equitable code review assignments across the
team.  Currently this migration is equivalent in the review assignments
it will create. Once stable, our expectation is that we will be able to
take advantage of PullApproves additional features for things like staged
reviews.
@IgorMinar IgorMinar added cla: yes action: merge The PR is ready for merge by the caretaker and removed cla: no labels Jan 28, 2020
@ngbot

ngbot Bot commented Jan 28, 2020

Copy link
Copy Markdown

I see that you just added the PR action: merge label, but the following checks are still failing:
    failure status "cla/google" is failing
    pending status "ci/circleci: setup" is pending
    pending missing required status "ci/circleci: build"
    pending missing required status "ci/circleci: lint"
    pending missing required status "ci/circleci: publish_snapshot"
    pending missing required status "ci/angular: size"
    pending 1 pending code review

If you want your PR to be merged, it has to pass all the CI checks.

If you can't get the PR to a green state due to flakes or broken master, please try rebasing to master and/or restarting the CI job. If that fails and you believe that the issue is not due to your change, please contact the caretaker and ask for help.

@googlebot

Copy link
Copy Markdown

A Googler has manually verified that the CLAs look good.

(Googler, please make sure the reason for overriding the CLA status is clearly documented in these comments.)

ℹ️ Googlers: Go here for more info.

@IgorMinar IgorMinar left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

I've rebased this and pushed a last set of fixup commits. I think this is good to go.

I'm marking this as "blocked" so that it is merged only once @josephperrott is around to babysit this PR and ensure that it doesn't break the world. 😄

@googlebot

Copy link
Copy Markdown

All (the pull request submitter and all commit authors) CLAs are signed, but one or more commits were authored or co-authored by someone other than the pull request submitter.

We need to confirm that all authors are ok with their commits being contributed to this project. Please have them confirm that by leaving a comment that contains only @googlebot I consent. in this pull request.

Note to project maintainer: There may be cases where the author cannot leave a comment, or the comment is not properly detected as consent. In those cases, you can manually confirm consent of the commit author(s), and set the cla label to yes (if enabled on your project).

ℹ️ Googlers: Go here for more info.

@googlebot

Copy link
Copy Markdown

A Googler has manually verified that the CLAs look good.

(Googler, please make sure the reason for overriding the CLA status is clearly documented in these comments.)

ℹ️ Googlers: Go here for more info.

josephperrott added a commit that referenced this pull request Jan 28, 2020
…34814)

We are migrating to PullApprove for our PR review management in an attempt
to allow for more granular and equitable code review assignments across the
team.  Currently this migration is equivalent in the review assignments
it will create. Once stable, our expectation is that we will be able to
take advantage of PullApproves additional features for things like staged
reviews.

PR Close #34814
@IgorMinar

Copy link
Copy Markdown
Contributor

live long and prosper pullaprove 🖖

gkalpak added a commit to gkalpak/angular that referenced this pull request Jan 29, 2020
This is a follow-up to angular#34814 to fix some typos in patterns and make
them more similar to the old patterns from `.github/CODEOWNERS`.
AndrewKushnir pushed a commit that referenced this pull request Jan 29, 2020
This is a follow-up to #34814 to fix some typos in patterns and make
them more similar to the old patterns from `.github/CODEOWNERS`.

PR Close #35015
AndrewKushnir pushed a commit that referenced this pull request Jan 29, 2020
This is a follow-up to #34814 to fix some typos in patterns and make
them more similar to the old patterns from `.github/CODEOWNERS`.

PR Close #35015
@angular-automatic-lock-bot

Copy link
Copy Markdown

This issue has been automatically locked due to inactivity.
Please file a new issue if you are encountering a similar or related problem.

Read more about our automatic conversation locking policy.

This action has been performed automatically by a bot.

@angular-automatic-lock-bot angular-automatic-lock-bot Bot locked and limited conversation to collaborators Feb 28, 2020
Sign up for free to subscribe to this conversation on GitHub. Already have an account? Sign in.

Labels

action: merge The PR is ready for merge by the caretaker area: build & ci Related the build and CI infrastructure of the project cla: yes target: patch This PR is targeted for the next patch release

Projects

None yet

Development

Successfully merging this pull request may close these issues.

6 participants