Skip to content

feat: remove @angular/http - #27038

Closed
CaerusKaru wants to merge 1 commit into
angular:masterfrom
CaerusKaru:adam/http
Closed

CaerusKaru wants to merge 1 commit into
angular:masterfrom
CaerusKaru:adam/http

Conversation

@CaerusKaru

@CaerusKaru CaerusKaru commented Nov 10, 2018 •

Copy link
Copy Markdown
Member

BREAKING CHANGE:

  • This removes @angular/http from the list of supported Angular core APIs. The new official API for HTTP in Angular is @angular/common/http

@CaerusKaru
CaerusKaru force-pushed the adam/http branch 18 times, most recently from c7533da to fb7d915 Compare November 13, 2018 04:30
@IgorMinar IgorMinar added this to the v8-candidates milestone Nov 15, 2018
@ngbot ngbot Bot removed this from the v8-candidates milestone Nov 15, 2018
@IgorMinar

Copy link
Copy Markdown
Contributor

I'd love for this to happen in v8! Would you be willing to work through the cleanup, writing up migration instructions for anyone still using the old http, and get this PR into a mergable state?

@IgorMinar IgorMinar added this to the v8-candidates milestone Nov 15, 2018
@ngbot ngbot Bot removed this from the v8-candidates milestone Nov 15, 2018
@IgorMinar IgorMinar added this to the v8-candidates milestone Nov 16, 2018
@ngbot ngbot Bot removed this from the v8-candidates milestone Nov 16, 2018
@IgorMinar IgorMinar added this to the v8-candidates milestone Nov 16, 2018
@ngbot ngbot Bot removed this from the v8-candidates milestone Nov 16, 2018
@IgorMinar IgorMinar added this to the v8-candidates milestone Nov 16, 2018
@ngbot ngbot Bot removed this from the v8-candidates milestone Nov 16, 2018
@CaerusKaru
CaerusKaru requested review from a team January 6, 2019 04:50
@CaerusKaru
CaerusKaru force-pushed the adam/http branch 2 times, most recently from 590f2ff to 5a182d2 Compare January 8, 2019 01:19
@CaerusKaru
CaerusKaru requested a review from a team January 13, 2019 18:18
@CaerusKaru CaerusKaru changed the title refactor: remove @angular/http build: remove @angular/http Jan 13, 2019
@CaerusKaru CaerusKaru changed the title build: remove @angular/http refactor: remove @angular/http Jan 13, 2019

@IgorMinar IgorMinar left a comment •

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

We are almost there!! Can you please fix the following:

git grep "@angular/http" | grep -v CHANGELOG.md


.github/CODEOWNERS:#  @angular/http
aio/tools/examples/shared/boilerplate/service-worker/package.json:    "@angular/http": "^7.0.0",
integration/platform-server/yarn.lock:"@angular/http@file:../../dist/packages-dist/http":
modules/playground/src/web_workers/worker-systemjs-configure.ts:    '@angular/http': '/packages-dist/http/bundles/http.umd.js',
packages/platform-server/package.json:    "@angular/http": "0.0.0-PLACEHOLDER",

Also, I looked into google3 and there is still a lot of usage of this package, so we'll need to figure out a plan for that. My current thinking is that unlike with other changes, we'll delete this package from github and schedule a google3 cleanup for after v8.

Lastly, similar to the DOCUMENT PR, please change the commit message type to "feat".

Comment thread packages/platform-server/src/location.ts Outdated
@CaerusKaru

Copy link
Copy Markdown
Member Author

@IgorMinar added remaining files, ready for another look!

@alexeagle

Copy link
Copy Markdown
Contributor

Could you please split it into two:

  1. remove any references to old http in our code, but still keep the implementation. This lets us still publish @angular/http for v8 if needed
  2. PR that actually deletes the package in case of victory

@CaerusKaru

Copy link
Copy Markdown
Member Author

@alexeagle I've opened #29355. We've already migrated all of the docs in AIO, and platform-server can't be migrated until we remove Http entirely (although technically we could: this is up to @IgorMinar).

@ankeshdave

Copy link
Copy Markdown

Will angular not support @angular/http in v8. Any clarity around this will help of people decide on migration strategy (I still have some projects were migrations is pending).

@petebacondarwin

Copy link
Copy Markdown
Contributor

@CaerusKaru - where are we at with this PR? It needs a rebase but it is not clear to me whether you successfully split the changes into two as requested by @alexeagle ?

@CaerusKaru

Copy link
Copy Markdown
Member Author

@petebacondarwin This PR hasn't been rebased on top of #29355, which is the split @alexeagle mentioned. Rebasing at this point would be tricky and I might just start from scratch, but only with commitment from the team that this is still needed.

@CaerusKaru

CaerusKaru commented Jan 22, 2020 •

Copy link
Copy Markdown
Member Author

Blocked on #34892 and #34814

@CaerusKaru CaerusKaru mentioned this pull request Jan 22, 2020
3 of 14 tasks
@kara

kara commented Apr 28, 2020

Copy link
Copy Markdown
Contributor

@CaerusKaru Rebase?

@kara kara left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

LGTM

@IgorMinar IgorMinar left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

LGTM! Thank you for being super patient!

The merging of this PR is still blocked on a config change CL in google3 that @matsko is landing right now.

Reviewed-for: global-approvers

@kara

kara commented May 4, 2020

Copy link
Copy Markdown
Contributor

G3 change should be in, so this is unblocked

@matsko

matsko commented May 4, 2020

Copy link
Copy Markdown
Contributor

@kara @IgorMinar unfortunately the G3 change is causing files to be removed. We can't merge this yet.

@matsko

matsko commented May 5, 2020

Copy link
Copy Markdown
Contributor

This is now unblocked because G3 doesn't sync @angular/http anymore.

Here's the latest sync (of this PR) that proves it: https://critique-ng.corp.google.com/cl/309988926

The legacy HTTP package was deprecated in v5 with the launch of
@angular/common/http. The legacy package hasn't been published
since v7, and will therefore not include a migration.
@angular-automatic-lock-bot

Copy link
Copy Markdown

This issue has been automatically locked due to inactivity.
Please file a new issue if you are encountering a similar or related problem.

Read more about our automatic conversation locking policy.

This action has been performed automatically by a bot.

Sign up for free to subscribe to this conversation on GitHub. Already have an account? Sign in.

Labels

action: merge The PR is ready for merge by the caretaker cla: yes effort2: days risk: medium target: major This PR is targeted for the next major release type: bug/fix

Projects

None yet

Development

Successfully merging this pull request may close these issues.

8 participants